Skip to content

Update module golang.org/x/sys to v0.48.0 (release-v1.36) - #1633

Merged
ske-prow[bot] merged 1 commit into
release-v1.36from
renovate/release-v1.36-golang.org-x-sys-0.x
Sep 16, 2026
Merged

ske-prow[bot] merged 1 commit into
release-v1.36from
renovate/release-v1.36-golang.org-x-sys-0.x

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
golang.org/x/sys v0.47.0v0.48.0 age confidence

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@ske-renovate-operator ske-renovate-operator Bot added the kind/upgrade Upgrading components or dependencies (e.g., by renovate) label Sep 8, 2026
@ske-renovate-operator
ske-renovate-operator Bot requested a review from a team September 8, 2026 12:32
@ske-prow ske-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 8, 2026
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/release-v1.36-golang.org-x-sys-0.x branch from c4f31a5 to 1c1381a Compare September 16, 2026 07:07
@ske-prow ske-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 16, 2026
@ske-prow

ske-prow Bot commented Sep 16, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: breuerfelix

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow

ske-prow Bot commented Sep 16, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 683102fa3d2f0636f66b280b5ee480fd05c23b0c

@ske-prow ske-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 16, 2026
@ske-prow
ske-prow Bot merged commit 723bbfd into release-v1.36 Sep 16, 2026
4 checks passed
@ske-prow
ske-prow Bot deleted the renovate/release-v1.36-golang.org-x-sys-0.x branch September 16, 2026 12:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/upgrade Upgrading components or dependencies (e.g., by renovate) lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant