chore(deps): update nx monorepo (major) - #808
Conversation
|
|
View your CI Pipeline Execution ↗ for commit 9a34d11 ☁️ Nx Cloud last updated this comment at |
4137139 to
1d638c5
Compare
1d638c5 to
9a34d11
Compare
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
This PR contains the following updates:
22.7.6→23.1.223.2.1(+2)22.7.6→23.1.223.2.1(+2)v4→v5Release Notes
nrwl/nx (@nx/express)
v23.1.2Compare Source
23.1.2 (2026-08-26)
🩹 Fixes
nx affected(#36569, #36568)@find(#36526)❤️ Thank You
v23.1.1Compare Source
23.1.1 (2026-07-30)
🚀 Features
🩹 Fixes
❤️ Thank You
v23.1.0Compare Source
23.1.0 (2026-07-13)
🚀 Features
🩹 Fixes
${configDir}in tsconfig path alias resolution (#36037, #35804)Angular v19 is no longer supported.
ESLint v8 is no longer supported. Nx requires ESLint v9
or later.
❤️ Thank You
v23.0.2Compare Source
23.0.2 (2026-07-10)
🚀 Features
🩹 Fixes
${configDir}in tsconfig path alias resolution (#36037, #35804)Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
This PR was generated by Mend Renovate. View the repository job log.